Skip to content

Clean up loose hostname matches - #52581

Merged
tbradsha merged 3 commits into
trunkfrom
fix/clean_up_loose_hostname_matches
Sep 22, 2026
Merged

tbradsha merged 3 commits into
trunkfrom
fix/clean_up_loose_hostname_matches

Conversation

@tbradsha

Copy link
Copy Markdown
Contributor

Proposed changes

This tightens some hostname matches.

Related product discussion/links

  • p1790010272417069-slack-C05Q5HSS013

Does this pull request change what data or activity we track or use?

Testing instructions

Ensure the code has no typos.

@tbradsha
tbradsha requested a review from a team September 21, 2026 21:21
@tbradsha tbradsha self-assigned this Sep 21, 2026
@tbradsha
tbradsha requested a review from a team as a code owner September 21, 2026 21:21
@tbradsha tbradsha added the [Status] Needs Review This PR is ready for review. label Sep 21, 2026
@github-actions

github-actions Bot commented Sep 21, 2026

Copy link
Copy Markdown
Contributor

Are you an Automattician? Please test your changes on all WordPress.com environments to help mitigate accidental explosions.

  • To test on WoA, go to the Plugins menu on a WoA dev site. Click on the "Upload" button and follow the upgrade flow to be able to upload, install, and activate the Jetpack Beta plugin. Once the plugin is active, go to Jetpack > Jetpack Beta, select your plugin (Jetpack or WordPress.com Site Helper), and enable the fix/clean_up_loose_hostname_matches branch.
  • To test on Simple, run the following command on your sandbox:
bin/jetpack-downloader test jetpack fix/clean_up_loose_hostname_matches
bin/jetpack-downloader test jetpack-mu-wpcom-plugin fix/clean_up_loose_hostname_matches

Interested in more tips and information?

  • In your local development environment, use the jetpack rsync command to sync your changes to a WoA dev blog.
  • Read more about our development workflow here: PCYsg-eg0-p2
  • Figure out when your changes will be shipped to customers here: PCYsg-eg5-p2

@github-actions github-actions Bot added [Boost Feature] Image CDN aka Photon [Feature] Masterbar WordPress.com Toolbar and Dashboard customizations [mu wpcom Feature] Wpcom Themes [Package] Jetpack mu wpcom WordPress.com Features [Package] Masterbar [Plugin] Boost A feature to speed up the site and improve performance. labels Sep 21, 2026
@github-actions

github-actions Bot commented Sep 21, 2026

Copy link
Copy Markdown
Contributor

Thank you for your PR!

When contributing to Jetpack, we have a few suggestions that can help us test and review your patch:

  • ✅ Include a description of your PR changes.
  • ✅ Add a "[Status]" label (In Progress, Needs Review, ...).
  • ✅ Add testing instructions.
  • ✅ Specify whether this PR includes any changes to data or privacy.
  • ✅ Add changelog entries to affected projects

This comment will be updated as you work on your PR and make changes. If you think that some of those checks are not needed for your PR, please explain why you think so. Thanks for cooperation 🤖


Follow this PR Review Process:

  1. Ensure all required checks appearing at the bottom of this PR are passing.
  2. Make sure to test your changes on all platforms that it applies to. You're responsible for the quality of the code you ship.
  3. You can use GitHub's Reviewers functionality to request a review.
  4. When it's reviewed and merged, you will be pinged in Slack to deploy the changes to WordPress.com simple once the build is done.

If you have questions about anything, reach out in #jetpack-developers for guidance!


Boost plugin:

No scheduled milestone found for this plugin.

If you have any questions about the release process, please ask in the #jetpack-releases channel on Slack.

@jp-launch-control

jp-launch-control Bot commented Sep 21, 2026

Copy link
Copy Markdown

Code Coverage Summary

Coverage changed in 1 file.

File Coverage Δ% Δ Uncovered
projects/plugins/boost/app/modules/optimizations/image-cdn/src/srcset.ts 53/56 (94.64%) -3.43% 2 ❤️‍🩹

Full summary · PHP report · JS report

@anomiex anomiex left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks reasonable, and as far as I can tell stuff still seems to work.

@tbradsha
tbradsha merged commit 69fec35 into trunk Sep 22, 2026
101 checks passed
@tbradsha
tbradsha deleted the fix/clean_up_loose_hostname_matches branch September 22, 2026 16:27
@github-actions github-actions Bot removed the [Status] Needs Review This PR is ready for review. label Sep 22, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

[Boost Feature] Image CDN aka Photon [Feature] Masterbar WordPress.com Toolbar and Dashboard customizations [mu wpcom Feature] Wpcom Themes [Package] Jetpack mu wpcom WordPress.com Features [Package] Masterbar [Plugin] Boost A feature to speed up the site and improve performance.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants