Skip to content

components: Improve regexes in DiffViewer component - #52582

Merged
anomiex merged 2 commits into
trunkfrom
fix/components-diffviewer-regexes
Sep 22, 2026
Merged

anomiex merged 2 commits into
trunkfrom
fix/components-diffviewer-regexes

Conversation

@anomiex

@anomiex anomiex commented Sep 21, 2026 •

Copy link
Copy Markdown
Contributor

Closes MONOREP-949

Proposed changes

A sequence like \s+(.*) can be inefficient if it winds up not matching, as it'll backtrack every possibility of whitespace matching the \s versus the .. We can fix it easily enough by doing like \s+(\S.*|), preventing the group from having leading whitespace.

The other regex here is pretty much impossible without atomic groups or possessive quantifiers. But since the intent is to get everything after a prefix, we can match just the prefix (which can be done efficiently) and then use .substring() to extract everything after.

Related product discussion/links

p1790010272417069-slack-C05Q5HSS013

Does this pull request change what data or activity we track or use?

No

Testing instructions

  • Should be able to use the Storybook to test the component. In projects/js-packages/storybook, run pnpm run storybook:dev.

A sequence like `\s+(.*)` can be inefficient if it winds up not
matching, as it'll backtrack every possibility of whitespace matching
the `\s` versus the `.`. We can fix it easily enough by doing like
`\s+(\S.*|)`, preventing the group from having leading whitespace.

The other regex here is pretty much impossible without atomic groups or
possessive quantifiers. But since the intent is to get everything after
a prefix, we can match just the prefix (which can be done efficiently)
and then use `.substring()` to extract everything after.
@anomiex
anomiex requested a review from a team September 21, 2026 21:39
@anomiex anomiex self-assigned this Sep 21, 2026
@anomiex anomiex added [Status] Needs Review This PR is ready for review. [Pri] Normal labels Sep 21, 2026
@github-actions

github-actions Bot commented Sep 21, 2026 •

Copy link
Copy Markdown
Contributor

Are you an Automattician? Please test your changes on all WordPress.com environments to help mitigate accidental explosions.

  • To test on WoA, go to the Plugins menu on a WoA dev site. Click on the "Upload" button and follow the upgrade flow to be able to upload, install, and activate the Jetpack Beta plugin. Once the plugin is active, go to Jetpack > Jetpack Beta, select your plugin (Jetpack or WordPress.com Site Helper), and enable the fix/components-diffviewer-regexes branch.
  • To test on Simple, run the following command on your sandbox:
bin/jetpack-downloader test jetpack fix/components-diffviewer-regexes
bin/jetpack-downloader test jetpack-mu-wpcom-plugin fix/components-diffviewer-regexes

Interested in more tips and information?

  • In your local development environment, use the jetpack rsync command to sync your changes to a WoA dev blog.
  • Read more about our development workflow here: PCYsg-eg0-p2
  • Figure out when your changes will be shipped to customers here: PCYsg-eg5-p2

@github-actions

Copy link
Copy Markdown
Contributor

Thank you for your PR!

When contributing to Jetpack, we have a few suggestions that can help us test and review your patch:

  • ✅ Include a description of your PR changes.
  • ✅ Add a "[Status]" label (In Progress, Needs Review, ...).
  • ✅ Add testing instructions.
  • ✅ Specify whether this PR includes any changes to data or privacy.
  • ✅ Add changelog entries to affected projects

This comment will be updated as you work on your PR and make changes. If you think that some of those checks are not needed for your PR, please explain why you think so. Thanks for cooperation 🤖


Follow this PR Review Process:

  1. Ensure all required checks appearing at the bottom of this PR are passing.
  2. Make sure to test your changes on all platforms that it applies to. You're responsible for the quality of the code you ship.
  3. You can use GitHub's Reviewers functionality to request a review.
  4. When it's reviewed and merged, you will be pinged in Slack to deploy the changes to WordPress.com simple once the build is done.

If you have questions about anything, reach out in #jetpack-developers for guidance!

@jp-launch-control

jp-launch-control Bot commented Sep 21, 2026 •

Copy link
Copy Markdown

Code Coverage Summary

This PR did not change code coverage!

That could be good or bad, depending on the situation. Everything covered before, and still is? Great! Nothing was covered before? Not so great. 🤷

Full summary · PHP report · JS report

tbradsha
tbradsha previously approved these changes Sep 22, 2026

@tbradsha tbradsha left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Looks great!


// Diff index
const header = /^(?:Index:|diff(?: -r \w+)+)\s+(.+?)\s*$/.exec( line );
const header = /^(?:Index:|diff(?: -r \w+)+)\s+/.exec( line );

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

I was playing with this a bit, and it seems an elegant solution.

I'll note this seems to be expecting a very specific diff format; it doesn't work with a standard git diff and the component actually crashes. Not sure it's worth addressing here.

Also, it seems the ensuing index.index is never actually used, so we could probably get away with removing this logic altogether, but again, that's perhaps out of scope.

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

As long as it crashes the same way with trunk, I'm not concerned about it for this PR. 😀

Following a comment above to where they got this from suggests that the diff -r something -r something format is for Mercurial. 🤷

Also, it seems the ensuing index.index is never actually used

Yeah, I noticed that too. I agree with "out of scope".

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

When I tried pasting in a git diff, it didn't crash. Worked fine. 🤷

*/
function parseFileHeader( index: Index ) {
const fileHeader = /^(---|\+\+\+)\s+(.*)\r?$/.exec( diffstr[ i ] );
const fileHeader = /^(---|\+\+\+)\s+(\S.*?|)\r?$/.exec( diffstr[ i ] );

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Shouldn't hurt anything, but why do we need the ? here?

Copy link
Copy Markdown
Contributor Author

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Hmm. We don't really, I'll take that out.

In PCRE the .* would match a \r making the \r? pointless, but JS . excludes \r, \u2028, and \u2029 as well as \n.

JS regex, unlike PCRE, excludes `\r` from `.` as well as `\n`.
@anomiex
anomiex merged commit b08ca5f into trunk Sep 22, 2026
120 checks passed
@anomiex
anomiex deleted the fix/components-diffviewer-regexes branch September 22, 2026 17:30
@github-actions github-actions Bot removed the [Status] Needs Review This PR is ready for review. label Sep 22, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants