Skip to content

chore: update CODEOWNERS to include additional owners - #705

Merged
runejo merged 1 commit into
mainfrom
chore-update-CODEOWNERS-to-include-additional-owners
Sep 23, 2026
Merged

runejo merged 1 commit into
mainfrom
chore-update-CODEOWNERS-to-include-additional-owners

Conversation

@runejo

@runejo runejo commented Sep 18, 2026

Copy link
Copy Markdown
Member

No description provided.

@sonarqubecloud

Copy link
Copy Markdown

@JohannesFinsveen JohannesFinsveen left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

👍

@runejo

runejo commented Sep 22, 2026

Copy link
Copy Markdown
Member Author

Waiting for SCB to also approve. Maybe @KentMossback should be added also?

@runejo
runejo merged commit ea5101c into main Sep 23, 2026
12 checks passed
@runejo
runejo deleted the chore-update-CODEOWNERS-to-include-additional-owners branch September 23, 2026 14:41
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants