Skip to content

Bump actions/setup-java from 5 to 6 - #965

Merged
lewismc merged 1 commit into
masterfrom
dependabot/github_actions/actions/setup-java-6
Sep 4, 2026
Merged

Bump actions/setup-java from 5 to 6#965
lewismc merged 1 commit into
masterfrom
dependabot/github_actions/actions/setup-java-6

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 1, 2026

Copy link
Copy Markdown
Contributor

Bumps actions/setup-java from 5 to 6.

Release notes

Sourced from actions/setup-java's releases.

v6.0.0

What's Changed

... (truncated)

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [actions/setup-java](https://github.com/actions/setup-java) from 5 to 6.
- [Release notes](https://github.com/actions/setup-java/releases)
- [Commits](actions/setup-java@v5...v6)

---
updated-dependencies:
- dependency-name: actions/setup-java
  dependency-version: '6'
  dependency-type: direct:production
  update-type: version-update:semver-major
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code labels Sep 1, 2026
@github-actions

github-actions Bot commented Sep 1, 2026

Copy link
Copy Markdown

Apache Yetus test-patch report

+1 overall
Vote Subsystem Runtime Log Comment
Prechecks
+1 dupname 0m 0s No case conflicting files found.
+1 @author 0m 0s The patch does not contain any @author tags.
master Compile Tests
Patch Compile Tests
+1 codespell 0m 2s No new issues.
+1 detsecrets 0m 32s No new issues.
+1 blanks 0m 0s The patch has no blanks issues.
+1 yamllint 0m 0s No new issues.
Other Tests
+1 asflicense 0m 1s The patch does not generate ASF License warnings.
0m 40s
Subsystem Report/Notes
GITHUB PR #965
Optional Tests dupname asflicense codespell detsecrets yamllint
uname Linux 97165ed3bdd1 6.17.0-1022-azure #22-Ubuntu SMP Mon Jul 27 17:24:03 UTC 2026 x86_64 x86_64 x86_64 GNU/Linux
Build tool nobuild
Personality /github/workspace/.yetus/personality.sh
git revision master / 6af373d
modules C: . U: .
versions git=2.34.1 codespell=2.4.1 detsecrets=1.5.0 yamllint=1.35.1
Powered by Apache Yetus 0.15.1 https://yetus.apache.org

This message was automatically generated.

@sonarqubecloud

sonarqubecloud Bot commented Sep 1, 2026

Copy link
Copy Markdown

@lewismc
lewismc merged commit 2549637 into master Sep 4, 2026
17 checks passed
@lewismc
lewismc deleted the dependabot/github_actions/actions/setup-java-6 branch September 4, 2026 06:12
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file github_actions Pull requests that update GitHub Actions code

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant