Skip to content

chore(deps): bump org.javassist:javassist from 3.30.2-GA to 3.33.0-GA - #1052

Open
dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/maven/org.javassist-javassist-3.33.0-GA
Open

dependabot[bot] wants to merge 1 commit into
mainfrom
dependabot/maven/org.javassist-javassist-3.33.0-GA

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Sep 6, 2026

Copy link
Copy Markdown
Contributor

Bumps org.javassist:javassist from 3.30.2-GA to 3.33.0-GA.

Release notes

Sourced from org.javassist:javassist's releases.

Javassist-3.33.0-GA

What's Changed

Full Changelog: jboss-javassist/javassist@rel_3_32_0_ga...rel_3_33_0_ga

Javassist-3.32.0-GA

  • Excludes javassist.tools.{reflect,rmi,web}. They are now included in ./examples/src/main.

  • Disables javassist.runtime.Desc.useContextClassLoader. Please use javassist.runtime.Desc.setUseContextClassLoaderLocally() instead.

Javassist 3.31.0-GA

What's Changed

New Contributors

Full Changelog: jboss-javassist/javassist@rel_3_30_2_ga...rel_3_31_0_ga

Changelog

Sourced from org.javassist:javassist's changelog.

Changes

version 3.34

version 3.33 on August 23, 2026

  • API changes suggested by @waydeshi and @lucianjohnhouse

  • GitHub PR #523

version 3.32 on June 21, 2026

  • API change suggested by Wayde Shi (GitHub: @waydeshi).

  • Excludes javassist.tools.{reflect,rmi,web}. They are now included in ./examples/src/main.

  • Disables javassist.runtime.Desc.useContextClassLoader

version 3.31.0 on April 20, 2026

version 3.30.2 on December 25, 2023

  • GitHub PR #473, 475, 476

version 3.30.1 on December 17, 2023

version 3.30 on December 17, 2023

  • GitHub PR #434, 448, 463 (Issue #462), 466, 467, 468, 469, 470,

version 3.29.2 on September 14, 2022

  • GitHub Issue #427.

version 3.29.1 on August 11, 2022

  • GitHub Issue #423.

  • Readme.html was deleted (GitHub Issue #414).

version 3.29 on May 13, 2022

... (truncated)

Commits

Dependabot compatibility score

Dependabot will resolve any conflicts with this PR as long as you don't alter it yourself. You can also trigger a rebase manually by commenting @dependabot rebase.


Dependabot commands and options

You can trigger Dependabot actions by commenting on this PR:

  • @dependabot rebase will rebase this PR
  • @dependabot recreate will recreate this PR, overwriting any edits that have been made to it
  • @dependabot show <dependency name> ignore conditions will show all of the ignore conditions of the specified dependency
  • @dependabot ignore this major version will close this PR and stop Dependabot creating any more for this major version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this minor version will close this PR and stop Dependabot creating any more for this minor version (unless you reopen the PR or upgrade to it yourself)
  • @dependabot ignore this dependency will close this PR and stop Dependabot creating any more for this dependency (unless you reopen the PR or upgrade to it yourself)

Bumps [org.javassist:javassist](https://github.com/jboss-javassist/javassist) from 3.30.2-GA to 3.33.0-GA.
- [Release notes](https://github.com/jboss-javassist/javassist/releases)
- [Changelog](https://github.com/jboss-javassist/javassist/blob/master/Changes.md)
- [Commits](https://github.com/jboss-javassist/javassist/commits)

---
updated-dependencies:
- dependency-name: org.javassist:javassist
  dependency-version: 3.33.0-GA
  dependency-type: direct:production
  update-type: version-update:semver-minor
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file java Pull requests that update java code labels Sep 6, 2026
@dependabot dependabot Bot added dependencies Pull requests that update a dependency file java Pull requests that update java code labels Sep 6, 2026
@openshift-ci

openshift-ci Bot commented Sep 6, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: dependabot[bot]

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 6, 2026

Copy link
Copy Markdown

Hi @dependabot[bot]. Thanks for your PR.

I'm waiting for a eclipse-che member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work.

Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies Pull requests that update a dependency file java Pull requests that update java code needs-ok-to-test

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants