Skip to content

chore(flake/nixpkgs): c043004d -> dc5d91f8 - #398

Merged
fredclausen merged 1 commit into
mainfrom
update-nixpkgs
Sep 8, 2026
Merged

chore(flake/nixpkgs): c043004d -> dc5d91f8#398
fredclausen merged 1 commit into
mainfrom
update-nixpkgs

Conversation

@fredclausen

Copy link
Copy Markdown
Member
Commit Message
5a1a5ef0 nixosTests.dawarich: allocate more memory
63077771 fcitx5-chewing: 5.1.12 -> 5.1.13
a13a918c maintainers/README: emphasize how to join the nixpkgs-maintainers team
b6c9cb05 linux_zen: 7.1.10 -> 7.2.3
8cd21efb linux_testing: 7.3-rc1 -> 7.3-rc2
0abd9e65 python3Packages.deepmerge: 3.0 -> 3.0.1
6df5502a python3Packages.griffelib: 2.2.0 -> 2.3.0
47669945 terraform-providers.vancluever_acme: 3.0.0 -> 3.1.0
2a9789b6 terraform-providers.hashicorp_tls: 4.3.0 -> 4.4.0
b3d92011 luau: 0.735 -> 0.737
39119d5c postfix-tlspol: 1.13.0 -> 1.14.0
f2546cdc python3Packages.tokenizers: 0.23.1 -> 0.23.2
f01e4679 python3Packages.types-appdirs: shorten description
f676d865 terraform-providers.argoproj-labs_argocd: 7.16.0 -> 7.17.0
688f4601 dawarich: 1.14.3 -> 1.14.4
679283fd python3Packages.pymap3d: init at 3.2.0
2ee11b66 python3Packages.scikit-survival: fix tests.full eval error
b4d5e034 python3Packages.huggingface-hub: 1.28.0 -> 1.30.0
7585015a font-awesome: use installFonts
c457037b libtlsrpt: 0.5.0 -> 0.5.1
393df32f overpass: use installFonts hook
5d54be40 far2l: 2.8.0 -> 2.9.0
698819ac python3Packages.pandera: enabel __darwinAllowLocalNetworking to fix tests on darwin
40e5b8c6 visidata: cleanup
c4b2e364 git-dumper: 1.0.8 -> 1.0.9
5d52a26a libretro.fmsx: 0-unstable-2026-06-04 -> 0-unstable-2026-09-06
29372f22 python3Packages.oss2: 2.18.3 -> 2.19.1
42af1bcd mcp-server-{filesystem,memory,sequential-thinking}: migrate to typescript 7
53f8da72 linearicons-free: use installFonts
f59660d5 pffft: 0-unstable-2025-06-09 -> 1.1.0
c4d79b30 hunk: 0.20.1 -> 0.21.1
b479b2f4 pfetch-rs: add `versionCheckHook`
5f5c0d4c pfetch: use `finalAttrs` pattern
af87f7f1 libretro.gambatte: 0-unstable-2026-08-14 -> 0-unstable-2026-08-21
600cc9b6 python3Packages.openccu-loom-types: drop
19f1b826 cinnamon: fix license
b805081e SDL_Pango: drop
0133f9be gtk-frdp: 0-unstable-2026-04-24 -> 0-unstable-2026-07-24
2bdbfd9e libgcc: enable __structuredAttrs
668f52b1 convertx: migrate to typescript 7
af7a9938 vikunja: 2.5.0 -> 2.6.0
d91a239c opencode: 1.18.28 -> 1.18.29
705ae6fc openscadPackages.dotscad: init at 3.3
ebb6cf7c agent-browser: 0.27.0 -> 0.36.0
0437f787 vscode-extensions.charliermarsh.ruff: 2026.74.0 -> 2026.78.0
598d5f0f sessiond: 0.2.0 -> 0.2.2
763e0998 glaze: 8.2.0 -> 8.3.0
e776a87a pgschema: 1.12.4 -> 1.12.5
9d84a06b python3Packages.torchtitan: 0.2.2 -> 0.3.0
55fe011f python3Packages.torch-checkpointing: init at 0.1.0
98b14fda python3Packages.spmd-types: init at 0.2.5
b9b2f375 comrak: 0.54.0 -> 0.55.0
09f685e5 olympus-unwrapped: 26.08.24.03 -> 26.09.06.01
8899e200 corepack: enable `structuredAttrs` and `strictDeps`
5bfbad6f python3Packages.libknot: 3.5.7 -> 3.5.8
27f9775b atproto-goat: 0.2.3 -> 0.2.4
6ce9d7be prometheus-knot-exporter: 3.5.7 -> 3.5.8
362d8e86 simdjson: add `updateScript` and `pkgConfigModules`
636ba822 python3Packages.tika: enable tests
1b39f0af mistral-vibe: 2.24.5 -> 2.25.0
08ac5b5a dawarich: 1.14.2 -> 1.14.3
94a7b140 vscode-extensions.visualjj.visualjj: 0.33.6 -> 0.35.0
54e6d933 libretro.vba-next: 0-unstable-2026-07-21 -> 0-unstable-2026-08-31
0b3b796a bichon: 2.0.2 -> 2.0.3
a9e7531c python3Packages.torchinfo: modernize
18739ac3 radicle-explorer: 0-unstable-2026-08-28 -> 0-unstable-2026-09-03
65129a95 home-assistant-custom-components.homematicip_local: 2.10.0 -> 2.11.0
35912e39 python3Packages.openccu-loom-client: 2026.8.26 -> 2026.9.3
f5eca56b gosmee: 0.31.0 -> 0.32.0
ea31f400 sing-box-dashboard: init at 0-unstable-2026-09-03
0d010275 torrserver: 143 -> 144.1
2aacb561 badness: 0.20.0 -> 0.23.0
a05f6664 python3Packages.aiohomematic-config: 2026.8.0 -> 2026.8.1
3ceff327 python3Packages.aiohomematic: 2026.8.5 -> 2026.9.1
8aa19ad3 vimPlugins.tsc-nvim: migrate to typescript 7
0af243e4 mediaharbor: migrate to typescript 7
7db3ae6e qui: migrate to typescript 7
fca15353 home-assistant.python3Packages.pytest-homeassistant-custom-component: 0.13.363 -> 0.13.364
c51a262f home-assistant.tests.components: don't run pytest with -vvv
3589a940 python3Packages.homeassistant-stubs: 2026.9.0 -> 2026.9.1
6bb3f348 home-assistant-custom-components.tibber_local: 2026.9.0 -> 2026.9.1
59db47cc home-assistant-custom-components.solax_modbus: 2026.09.1 -> 2026.09.2
ac6d9e19 home-assistant-custom-components.polaris-mqtt: 1.1.5 -> 1.1.6
634c883d home-assistant-custom-components.llm_intents: 1.9.0 -> 1.10.0
44761484 home-assistant-custom-components.indego: 6.0.0 -> 6.1.1
df7837cb python3Packages.pyindego: 3.2.2 -> 3.2.3
bb1e3c3e home-assistant-custom-components.frigidaire: 0.1.38 -> 0.1.43
5249a546 home-assistant-custom-components.fellow: 2.0.0 -> 2.0.1
35e98159 home-assistant-custom-components.ecoflow_ble: 1.1.0 -> 1.1.1
67c6ff27 home-assistant-custom-components.daikin_onecta: 4.6.18 -> 4.6.19
92b54316 home-assistant-custom-components.blueprints-updater: 2.14.4 -> 2.14.5
09791a45 home-assistant: 2026.9.0 -> 2026.9.1
3f5b66aa fcitx5-hangul: 5.1.10 -> 5.1.11
d704f731 ironcalc: migrate to typescript 7
ff81bddb python3Packages.baselines: fix by using setuptools_80 as build-system
a836338d python3Packages.ddgs: 9.14.2 -> 9.16.0
497c9ea3 python3Packages.baselines: cleanup
7598a845 n98-magerun2: 10.0.1 -> 10.0.2
907f00fd raycast: 2.0.5.0 -> 2.2.0.0
a3989ee7 ignition: migrate to typescript 7
fc9be56f python3Packages.torchrl: skip failing test
4e6554b4 python3Packages.tensorflow{,-bin}: remove file that incorrectl sets RTLD_GLOBAL and causes symbol collisions
19bf3164 triton-llvm: set BUILD_SHARED_LIBS to save ~3GiB
064e2133 folia-major: 0.7.0 -> 0.7.3
ee6e3694 localsend: adjust desktop item
f1997474 vscode-extensions.ms-ceintl.vscode-language-pack-fr: 1.131.2026082318 -> 1.131.2026090407
3ba830f3 vesktop: update description and add comment
1f6ea975 tui-journal: 0.17.0 -> 0.18.0
ccff5725 python3Packages.jax: add cuda to meta.teams
7c972426 python3Packages.jax: passthru cleanup
91209ec2 python3Packages.jax: enable pytest-xdist
a91c6869 python3Packages.jax: fix cuda support with cuda 13
b7d71533 python3Packages.jax-cuda13-plugin: init at 0.11.1
3aca802d python3Packages.jax-cuda13-pjrt: init at 0.11.1
af2ecae3 tomat: 2.12.0 -> 2.13.0
7cd7f257 python3Packages.scs: add GaetanLepage to maintainers
3ca9cb90 scs: add GaetanLepage to maintainers
16c7832d scs: 3.3.0 -> 3.3.1
ab751c9b python3Packages.exa-py: 2.19.0 -> 2.20.0
6ff744c6 python3Packages.jax-cuda12-plugin: add cuda to meta.teams
619f0c6e python3Packages.jax-cuda12-plugin: add GaetanLepage to maintainers
f1b754cd python3Packages.jax-cuda12-plugin: use meta.problems to handle cuda compat
d7fd7991 python3Packages.jax-cuda12-plugin: cleanup postInstall
c3c18a36 python3Packages.jax-cuda12-pjrt: add cuda to meta.teams
b0a5dc75 python3Packages.jax-cuda12-pjrt: add GaetanLepage to maintainers
f7749df3 python3Packages.jax-cuda12-pjrt: use meta.problems to handle cuda compat
e85f7ece python3Packages.jax-cuda12-pjrt: cleanup postInstall
788e3aa2 nixfmt-rs: 0.4.1 -> 0.5.3
01b1e647 ctx7: 0.5.9 -> 0.5.10
bf510350 terraform-providers.sumologic_sumologic: 3.3.1 -> 3.3.2
9d11387b terraform-providers.spotinst_spotinst: 1.241.0 -> 1.241.1
ba755b3e vimPlugins.vim-be-better: init at 0-unstable-2025-06-21
a3d48d8d python3Packages.torchrl: use pytest-xdist to speed up tests
f17ee1e2 python3Packages.torchrl: 0.13.3 -> 0.14.0
38c03185 tinymist: 0.15.2 -> 0.15.6
508fbfd3 python3Packages.checkdmarc: 5.17.5 -> 6.0.1
778bbf39 nerva: 1.69.0 -> 1.69.5
7664fa84 jrl-cmakemodules: enable __structuredAttrs and strictDeps
dfa7f658 jrl-cmakemodules: add missing openmp checkInputs
a98156c4 shellshot: init at 0.6.2
2659c0d5 bootdev-cli: 1.32.1 -> 1.32.2
262ced67 python3Packages.detectron2: drop
dc7f656b python3Packages.microsoft-kiota-serialization-text: 1.12.0 -> 1.12.1
eb338272 python3Packages.microsoft-kiota-serialization-multipart: 1.12.0 -> 1.12.1
15278ea9 python3Packages.microsoft-kiota-serialization-json: 1.12.0 -> 1.12.1
547e7899 python3Packages.microsoft-kiota-serialization-form: 1.12.0 -> 1.12.1
1e383bad python3Packages.microsoft-kiota-http: 1.12.0 -> 1.12.1
0b134cb5 python3Packages.microsoft-kiota-authentication-azure: 1.12.0 -> 1.12.1
d6f035a4 python3Packages.microsoft-kiota-abstractions: 1.12.0 -> 1.12.1
23c7f360 aptakube: 1.19.3 -> 1.19.6
598f813f sasview: fix build
243cac63 terminator: 2.1.5 -> 2.1.6
5678b010 clapgrep: 26.02 -> 26.09
ec823ecc python3Packages.iamdata: 0.1.202609051 -> 0.1.202609061
5036baff vimPlugins.notmuch-nvim: init at 0.4.0
e362fb08 oxipng: 10.1.1 -> 10.2.1
0a688e58 libretro.o2em: 0-unstable-2026-04-20 -> 0-unstable-2026-07-21
844018b2 tabularis: 0.21.0 -> 0.22.0
71cf81d4 nixos/alertmanager-gotify-bridge: ProtectProc = true -> "invisible"
c9d9b49a brave, brave-origin: 1.94.117 -> 1.94.121
6c50bb54 python3Packages.prefect: 3.8.4 -> 3.8.5
a7a5d19b brave: replace update script
3f946833 gvm-libs: 23.9.3 -> 23.10.0
1cd763b9 gogup: 1.8.1 -> 1.9.1
ef1fb352 ggml: 0.22.0 -> 0.23.0
3eba0cd4 libretro.genesis-plus-gx: 0-unstable-2026-08-21 -> 0-unstable-2026-08-28
93c00f06 wmderlandc: unstable-2020-07-17 -> 1.0.5
ceaff50f wmderland: unstable-2020-07-17 -> 1.0.5
85fc8c40 rusthound-ce: 2.5.5 -> 2.5.9
fdbde5b4 homepage-dashboard: 2.1.2 -> 2.2.0
94ae41e9 pkgsCuda.suitesparse: fix build
43f22b2c python3Packages.types-appdirs: fix meta.homepage
2237e48d python3Packages.types-appdirs: modernize
549b22de python3Packages.types-appdirs: migrate to pyproject
b4f60cb8 python3Packages.torchinfo: migrate to pyproject
fae079ea python3Packages.timelib: move cython to build-system
658546b6 python3Packages.timelib: modernize
035045e0 python3Packages.timelib: migrate to pyproject
c6998ee8 python3Packages.tika: modernize
880223c3 python3Packages.tika: migrate to pyproject
16b87b3a phpExtensions.snuffleupagus: 0.13.0 -> 0.14.0
b4fc8c6d terraform-providers.hashicorp_google-beta: 7.44.0 -> 8.1.0
6f10f74c terraform-providers.aiven_aiven: 4.61.0 -> 4.62.0
435b6538 cargo-zigbuild: 0.23.2 -> 0.23.4
b7f01588 capacities: 1.68.2 -> 1.70.2
77fc06e4 serie: 0.8.2 -> 0.9.0
6b63e5ae plantuml-server: 1.2026.6 -> 1.2026.8
26c1a15d nest-cli: 11.0.18 -> 12.0.0
2a327aa4 rgrc: enable __structuredAttrs
f8f80112 jwtinfo: 0.6.1 -> 0.7.0
ee6d8be3 home-assistant.frontend: 20260826.4 -> 20260826.6
a52c96ab python3Packages.modbus-connection: fix tests with tmodbus 0.6.2
2545ff45 python3Packages.tmodbus: 0.6.1 -> 0.6.2
c4d4d371 python3Packages.serialx: 1.9.0 -> 1.10.0
25112e6b python3Packages.reolink-aio: 0.21.14 -> 0.21.15
6b9203ef python3Packages.python-roborock: 7.1.1 -> 7.2.3
e2656481 python3Packages.pylitterbot: 2025.6.4 -> 2025.6.5
3c232e48 python3Packages.pydaikin: 2.19.0 -> 2.19.1
b57a5de7 python3Packages.knx-frontend: 2026.8.28.62336 -> 2026.9.4.63549
21766a92 python3Packages.env-canada: 0.19.1 -> 0.19.2
9d4b9009 python3Packages.aioflo: use finalAttrs
207d9153 python3Packages.aioflo: 2021.11.0 -> 2026.09.3
d52a9752 grpc-health-probe: 0.4.56 -> 0.4.57
bf22633f verilator: 5.050 -> 5.052
ac026c72 python3Packages.py-cpuinfo2: 10.1.0 -> 10.1.1
24fa7ad1 vscode-extensions.quarto.quarto: 1.135.0 -> 1.137.0
2c21ac12 piper-tts: enable tests
40114c3e python3Packages.pyopenjtalk-plus: disable failing tests on aarch64-linux
f2707513 vitess: 24.0.2 -> 24.0.3
8471eadf bcal: 2.5 -> 2.6
dcf8fd1d lazygit: 0.64.1 -> 0.65.0
8b86db43 weaviate: 1.39.1 -> 1.39.2
69fcb7fb feh: 3.12.3 -> 3.12.4
b93a2959 piper-tts: 1.7.0 -> 1.8.0
059c5e78 python3Packages.rio-tiler: disable failing tests
ead7b803 npb: 1.0.0 -> 1.1.0
ed34526e icon-library: 0.0.19 -> 0.0.22
a4257a59 kener: 4.1.2 -> 4.1.4
e9d4e051 viewnior: Add avif, heic, and jxl image support
a4eebe5c terraform-providers.cloudposse_utils: 2.6.0 -> 2.7.0
64b9d620 wrangler: 4.126.0 -> 4.129.0
07a9b5eb prevail: 0.2.0 -> 0.2.6
209c4430 pi-coding-agent: 0.85.0 -> 0.85.1
96a4fb42 calcure: 3.3 -> 3.4
93ef0849 typescript: pin to typescript_7
896209a1 treewide: pin typescript to typescript_5
51bfc541 clickhouse-lts: 26.3.17.110-lts -> 26.8.2.7-lts
fb0d0397 typescript_7: rename from typescript-go
6e88497b mqtt-explorer: pin typescript to typescript_5
de189a2b weylus: pin typescript to typescript_5
cd15929c vscode-json-languageserver: pin typescript to typescript_5
99fd1996 vscode-css-languageserver: pin typescript to typescript_5
150d05df opcua-commander: pin typescript to typescript_5
4e1f377d mediaharbor: pin typescript to typescript_5
6dc86623 mcp-server-sequential-thinking: pin typescript to typescript_5
5d943d3f mcp-server-memory: pin typescript to typescript_5
89e71464 mcp-server-filesystem: pin typescript to typescript_5
71431b4e imgbrd-grabber: pin typescript to typescript_5
3beb3aab kdePackages.karousel: pin typescript to typescript_5
a7b611a5 gnome-weather: pin typescript to typescript_5
9dfa7089 github-desktop: pin typescript to typescript_5
c084674a decibels: pin typescript to typescript_5
ab63f21e convertx: pin typescript to typescript_5
3ad69572 ags_1: pin typescript to typescript_5
402a9f6b python3Packages.hdf5plugin: 7.0.0 -> 7.1.0
427e56ef kops: remove unused arguments
beca6d7c maintainers: fix jfr github handle
d337a59f krep: modernize
d65aa1e5 llama-cpp: 0.3.0 -> 0.4.0
4a6afa6e python3Packages.modelscope: mark CVE-2026-84202
c5c21ffc hockeypuck: 2.4 -> 2.4.1
986c97cb nixos/suricata: ProtectProc = true -> "invisible"
fa757497 sgt-puzzles: 20260720.3c36322 -> 20260905.43eefe8
55eb1a83 lagrange: 1.21.0 -> 1.21.1
59f40388 httrack: 3.49.24 -> 3.50.1
caeba43d python3Packages.mizani: 0.14.4 -> 0.14.5
a02431e6 uiua-unstable: 0.19.0-rc.1 -> 0.19.1
322e385b uiua: 0.18.1 -> 0.19.1
98df7034 ntopng: apply patch for CVE-2026-86091 CVE-2026-86090
d81b6abf maintainers: remove dwrege
22c20f97 typescript_5: rename from typescript, add alias
2de2c1d2 terraform-providers.hashicorp_helm: 3.2.0 -> 3.3.0
2106c16f cargo-binstall: 1.22.0 -> 1.23.0
09df271b maintainers/github-teams.json: Automated sync
9521c523 maintainers/README.md: update with automatic re-invite instructions
6ee6b5ad hamr: init at 1.1
eb53446b vimPlugins.avante-nvim: install avante executable
4c498efa audacious-plugins: Remove interim patches for libsidplay >= 3.0
80760b72 audacious-plugins: 4.5.1 -> 4.6.1
a5c9c7be llama-cpp-{cuda,rocm,vulkan}: add nixpkgs-update: no auto update
3ca9f996 neowall: 0.6.0 -> 0.6.1
0cded46f antigravity-acp: init at 1.0.0
7fc9ca16 beamPackages.ex_doc: 0.40.3 -> 0.40.4
6a872395 incus-compose: 1.3.1 -> 1.3.2
11f73219 docker: 29.7.2 -> 29.8.0
03a4c5e4 node-red: 5.0.4 -> 5.0.6
29f0b104 anchor: 1.1.2 -> 1.2.0
fa0e38b2 readeck: 0.23.1 -> 0.23.2
9febaad5 session-desktop: 1.18.0 -> 1.18.1
35970847 satisfactorymodmanager: 3.0.7 -> 3.1.0
853b1d6d blesh: 0.4.0-devel3-unstable-2026-08-19 -> 0.4.0-devel3-unstable-2026-09-05
2192b8be sandbox-runtime: 0.0.74 -> 0.0.75
2c7c31ba taproot-assets: 0.8.2 -> 0.8.3
dff673c0 lanraragi: 0.9.60 -> 0.9.81
1ef63e77 perlPackages.MojoliciousPluginOpenAPI: 5.09 -> 5.12
05db70f1 perlPackages.JSONValidator: 5.14 -> 5.19
08d40ce0 sentry-native: 0.16.2 -> 0.16.5
52181013 gh: 2.99.0 -> 2.100.0
a03897ed weechat: 4.10.0 -> 4.10.1
fb32625f qmapshack: 1.21.0 -> 1.21.1
5a1a5ef0 nixosTests.dawarich: allocate more memory
63077771 fcitx5-chewing: 5.1.12 -> 5.1.13
a13a918c maintainers/README: emphasize how to join the nixpkgs-maintainers team
b6c9cb05 linux_zen: 7.1.10 -> 7.2.3
8cd21efb linux_testing: 7.3-rc1 -> 7.3-rc2
0abd9e65 python3Packages.deepmerge: 3.0 -> 3.0.1
6df5502a python3Packages.griffelib: 2.2.0 -> 2.3.0
47669945 terraform-providers.vancluever_acme: 3.0.0 -> 3.1.0
2a9789b6 terraform-providers.hashicorp_tls: 4.3.0 -> 4.4.0
b3d92011 luau: 0.735 -> 0.737
39119d5c postfix-tlspol: 1.13.0 -> 1.14.0

... and 85 more commits (truncated)

Copilot AI lite review requested due to automatic review settings September 8, 2026 02:17
@fredclausen
fredclausen enabled auto-merge (rebase) September 8, 2026 02:17

Copilot AI left a comment

Copy link
Copy Markdown

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Copilot wasn't able to review any files in this pull request.


💡 Add a code-review agent skill or configure MCP servers for context-aware, tailored reviews. Learn more in the docs.

@coderabbitai

coderabbitai Bot commented Sep 8, 2026

Copy link
Copy Markdown

Important

Review skipped

Review was skipped due to path filters

⛔ Files ignored due to path filters (1)
  • flake.lock is excluded by !**/*.lock

CodeRabbit blocks several paths by default. You can override this behavior by explicitly including those paths in the path filters. For example, including **/dist/** will override the default block on the dist directory, by removing the pattern from both the lists.

⚙️ Run configuration

Configuration used: Organization UI

Review profile: CHILL

Plan: Team

Run ID: 178a2aa7-908b-4a69-9d3d-11c1a0485c3e

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@fredclausen
fredclausen merged commit 53328c1 into main Sep 8, 2026
2 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants