Skip to content

chore(dev-server): migrate to pure esm - #358

Merged
yusukebe merged 2 commits into
honojs:mainfrom
ryuapp:chore/migrate-to-pure-esm
May 23, 2026
Merged

yusukebe merged 2 commits into
honojs:mainfrom
ryuapp:chore/migrate-to-pure-esm

Conversation

@ryuapp

@ryuapp ryuapp commented May 7, 2026

Copy link
Copy Markdown
Contributor

There is usually no need for Vite plugins to support CommonJS, so we are dropping it to reduce bundle size.

If there are special circumstances, require(esm) has been supported since Node.js v20.19.0, so you might want to use that (though it seems unnecessary).

@changeset-bot

changeset-bot Bot commented May 7, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 7a1f00d

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
@hono/vite-dev-server Minor

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@yusukebe yusukebe left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

LGTM!

@yusukebe

Copy link
Copy Markdown
Member

@ryuapp You are right. We don't need CommonJS. Thanks!

@yusukebe
yusukebe merged commit ff50c57 into honojs:main May 23, 2026
4 checks passed
@github-actions github-actions Bot mentioned this pull request May 23, 2026
@ryuapp
ryuapp deleted the chore/migrate-to-pure-esm branch May 23, 2026 06:24
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants