Skip to content

i18n(dde-cooperation): Updates for project Deepin Desktop Environment - #785

Merged
lzwind merged 3 commits into
masterfrom
transifex_update_817060b5c4
Sep 20, 2026
Merged

lzwind merged 3 commits into
masterfrom
transifex_update_817060b5c4

Conversation

@transifex-integration

@transifex-integration transifex-integration Bot commented Sep 7, 2026 •

Copy link
Copy Markdown
Contributor

The following localization files have been updated:

Details

Parameter Value
Transifex Project Deepin Desktop Environment
Grouping All Resources and Languages

Files

Source File Translation File Resource Language Type Mode
translations/dde-cooperation-desktop/desktop.ts translations/dde-cooperation-desktop/desktop_pl.ts translations/dde-cooperation-desktop/desktop.ts (master) pl TRANSLATION default
translations/dde-cooperation/dde-cooperation.ts translations/dde-cooperation/dde-cooperation_pl.ts translations/dde-cooperation/dde-cooperation.ts (master) pl TRANSLATION default

100% translated source file: 'desktop.ts'
on 'pl'.
100% translated source file: 'dde-cooperation.ts'
on 'pl'.
@deepin-ci-robot

Copy link
Copy Markdown

Hi @transifex-integration[bot]. Thanks for your PR.

I'm waiting for a linuxdeepin member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

100% translated source file: 'dde-cooperation.ts'
on 'pl'.
@lzwind
lzwind merged commit e28a164 into master Sep 20, 2026
25 of 26 checks passed
@deepin-ci-robot

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: lzwind, transifex-integration[bot]

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@re2zero
re2zero deleted the transifex_update_817060b5c4 branch September 20, 2026 07:49
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants