Tnc7.0 release 4.22 - #971
openshift-merge-bot[bot] merged 1 commit into
Conversation
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Organization UI Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
| clusterImageSetNameRef: "openshift-4.18-disconnected" | ||
| sshPublicKey: "ssh-rsa AAAA...." | ||
| name: "assisted-deployment-pull-secret" | ||
| clusterImageSetNameRef: "openshift-4.18-disconnected" |
There was a problem hiding this comment.
@a1bokhari I think this should be 4.22?
There was a problem hiding this comment.
Changed. Thanks1
|
@snecklifter: changing LGTM is restricted to collaborators DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
| namespace: openshift-logging | ||
| # Update with specific target version | ||
| name: cluster-logging.v6.2.11 | ||
| name: cluster-logging.v6.5.1 |
There was a problem hiding this comment.
I believe this needs to be 6.6 for OpenShift 4.22
|
@snecklifter: changing LGTM is restricted to collaborators DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
| namespace: openshift-logging | ||
| # Update with specific target version | ||
| name: cluster-logging.v6.2.11 | ||
| name: cluster-logging.v6.6.0 |
There was a problem hiding this comment.
nit: Should probably be cluster-logging.v6.6.1 really
|
@snecklifter: changing LGTM is restricted to collaborators DetailsIn response to this: Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. |
There was a problem hiding this comment.
Requesting to delete this one since the upgrade guide can be referenced in a different location.
There was a problem hiding this comment.
Requesting to delete this one since the upgrade guide can be referenced in a different location.
There was a problem hiding this comment.
should we remove this test namespace CR?
Pick up 4.22 RDS submodules TNC 7.0 Change cluster image to 4.22 - per snecklifter review comment update openshift-logging and loki to 6.6 Fix MCP pausing for MGMT cluster upgrade changing logging z-stream to 6.6.1 remove upgrade docs. Content moved to TNC 7.0 upgrade guide remove test ns
df0e006 to
9af6db2
Compare
|
[APPROVALNOTIFIER] This PR is APPROVED This pull-request has been approved by: a1bokhari, clindo84 The full list of commands accepted by this bot can be found here. The pull request process is described here DetailsNeeds approval from an approver in each of these files:
Approvers can indicate their approval by writing |
c2a5b92
into
openshift-kni:tnc7.0-release-4.22
No description provided.