Skip to content

[release-4.16] OCPBUGS-103070,OCPBUGS-103065: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3 - #1367

Open
sunku5494 wants to merge 1 commit into
openshift:release-4.16from
sunku5494:CVE-2026-33814-4.16
Open

[release-4.16] OCPBUGS-103070,OCPBUGS-103065: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3#1367
sunku5494 wants to merge 1 commit into
openshift:release-4.16from
sunku5494:CVE-2026-33814-4.16

Conversation

@sunku5494

Copy link
Copy Markdown

Addressing CVE-2026-33814

@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci-robot openshift-ci-robot added jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. labels Aug 31, 2026
@coderabbitai

coderabbitai Bot commented Aug 31, 2026

Copy link
Copy Markdown

Important

Review skipped

Auto reviews are disabled on base/target branches other than the default branch.

Please check the settings in the CodeRabbit UI or the .coderabbit.yaml file in this repository. To trigger a single review, invoke the @coderabbitai review command.

⚙️ Run configuration

Configuration used: Repository: openshift/coderabbit/.coderabbit.yaml

Review profile: CHILL

Plan: Enterprise

Run ID: ce13c34c-7ecc-4cbd-add0-037b92374842

You can disable this status message by setting the reviews.review_status to false in the CodeRabbit configuration file.

Use the checkbox below for a quick retry:

  • 🔍 Trigger review

Comment @coderabbitai help to get the list of available commands.

@openshift-ci-robot

Copy link
Copy Markdown

@sunku5494: This pull request references Jira Issue OCPBUGS-103062, which is invalid:

  • expected dependent Jira Issue OCPBUGS-103266 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ASSIGNED instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Addressing CVE-2026-33814

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@openshift-ci
openshift-ci Bot requested review from oceanc80 and perdasilva August 31, 2026 07:41
@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: sunku5494
Once this PR has been reviewed and has the lgtm label, please assign kevinrizza for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

…com/openshift-sustaining/net@v0.35.0-sec.3

Signed-off-by: Venkata Charan Sunku <vsunku@redhat.com>
@sunku5494
sunku5494 force-pushed the CVE-2026-33814-4.16 branch from 1e70e57 to 654d965 Compare August 31, 2026 07:44
@sunku5494 sunku5494 changed the title [release-4.16] OCPBUGS-103062: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3 [release-4.16] OCPBUGS-103070: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3 Aug 31, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@sunku5494: This pull request references Jira Issue OCPBUGS-103070, which is invalid:

  • expected dependent Jira Issue OCPBUGS-103286 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ASSIGNED instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Addressing CVE-2026-33814

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

@prabhapa

Copy link
Copy Markdown

/lgtm

@openshift-ci openshift-ci Bot added the lgtm Indicates that a PR is ready to be merged. label Aug 31, 2026
@openshift-merge-bot

Copy link
Copy Markdown
Contributor

Scheduling tests matching the pipeline_run_if_changed or not excluded by pipeline_skip_if_only_changed parameters:
/test e2e-gcp-console-olm
/test e2e-gcp-olm
/test e2e-gcp-ovn
/test e2e-upgrade

@openshift-ci

openshift-ci Bot commented Aug 31, 2026

Copy link
Copy Markdown
Contributor

@sunku5494: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@jkaurredhat

Copy link
Copy Markdown

/label backport-risk-assessed
/label approved

@openshift-ci openshift-ci Bot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 31, 2026
@MrSanketkumar

Copy link
Copy Markdown

/retitle [release-4.16] OCPBUGS-103070,OCPBUGS-103065: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3

@openshift-ci openshift-ci Bot changed the title [release-4.16] OCPBUGS-103070: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3 [release-4.16] OCPBUGS-103070,OCPBUGS-103065: Replace golang.org/x/net with github.com/openshift-sustaining/net@v0.35.0-sec.3 Sep 7, 2026
@openshift-ci-robot

Copy link
Copy Markdown

@sunku5494: This pull request references Jira Issue OCPBUGS-103070, which is invalid:

  • expected dependent Jira Issue OCPBUGS-103286 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is ASSIGNED instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

This pull request references Jira Issue OCPBUGS-103065, which is invalid:

  • expected dependent Jira Issue OCPBUGS-103304 to be in one of the following states: VERIFIED, RELEASE PENDING, CLOSED (ERRATA), CLOSED (CURRENT RELEASE), CLOSED (DONE), CLOSED (DONE-ERRATA), but it is New instead

Comment /jira refresh to re-evaluate validity if changes to the Jira bug are made, or edit the title of this pull request to link to a different bug.

The bug has been updated to refer to the pull request using the external bug tracker.

Details

In response to this:

Addressing CVE-2026-33814

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. jira/invalid-bug Indicates that a referenced Jira bug is invalid for the branch this PR is targeting. jira/severity-important Referenced Jira bug's severity is important for the branch this PR is targeting. jira/valid-reference Indicates that this PR references a valid Jira ticket of any type. lgtm Indicates that a PR is ready to be merged.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

5 participants